Skip to content

python3Packages.lifelines: cleanup, fix - #563953

Merged
pbsds merged 1 commit into
NixOS:masterfrom
GaetanLepage:fix/lifelines
Sep 16, 2026
Merged

pbsds merged 1 commit into
NixOS:masterfrom
GaetanLepage:fix/lifelines

Conversation

@GaetanLepage

@GaetanLepage GaetanLepage commented Sep 16, 2026

Copy link
Copy Markdown
Contributor

Things done

@nixpkgs-ci
nixpkgs-ci Bot requested a review from swflint September 16, 2026 15:58
@nixpkgs-ci nixpkgs-ci Bot added 10.rebuild-linux: 1-10 This PR causes between 1 and 10 packages to rebuild on Linux. 10.rebuild-darwin: 1-10 This PR causes between 1 and 10 packages to rebuild on Darwin. 6.topic: python Python is a high-level, general-purpose programming language. labels Sep 16, 2026

@swflint swflint left a comment

Copy link
Copy Markdown
Contributor

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

Given all tests pass, I'm happy to approve.

@GaetanLepage

Copy link
Copy Markdown
Contributor Author

nixpkgs-review result

Generated using nixpkgs-review.

Command: nixpkgs-review pr 563953
Commit: 6750d8f5ad7468976f805863ac73387de438fa6b


x86_64-linux

✅ 2 packages built:
  • python314Packages.lifelines
  • python313Packages.lifelines

aarch64-linux

✅ 2 packages built:
  • python314Packages.lifelines
  • python313Packages.lifelines

aarch64-darwin

✅ 2 packages built:
  • python314Packages.lifelines
  • python313Packages.lifelines

@nixpkgs-ci nixpkgs-ci Bot added 12.approvals: 1 This PR was reviewed and approved by one person. 12.approved-by: package-maintainer This PR was reviewed and approved by a maintainer listed in any of the changed packages. labels Sep 16, 2026
@nixpkgs-ci nixpkgs-ci Bot added 12.approvals: 2 This PR was reviewed and approved by two persons. and removed 12.approvals: 1 This PR was reviewed and approved by one person. labels Sep 16, 2026
@GaetanLepage
GaetanLepage requested a review from pbsds September 16, 2026 20:12
@pbsds
pbsds added this pull request to the merge queue Sep 16, 2026
Merged via the queue into NixOS:master with commit 2a959b3 Sep 16, 2026
35 checks passed
@GaetanLepage
GaetanLepage deleted the fix/lifelines branch September 16, 2026 20:35
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

6.topic: python Python is a high-level, general-purpose programming language. 10.rebuild-darwin: 1-10 This PR causes between 1 and 10 packages to rebuild on Darwin. 10.rebuild-linux: 1-10 This PR causes between 1 and 10 packages to rebuild on Linux. 12.approvals: 2 This PR was reviewed and approved by two persons. 12.approved-by: package-maintainer This PR was reviewed and approved by a maintainer listed in any of the changed packages.

Projects

None yet

Development

Successfully merging this pull request may close these issues.

4 participants