Skip to content

zed-editor: 1.19.2 -> 1.20.1 - #563960

Merged
GaetanLepage merged 1 commit into
NixOS:masterfrom
schembriaiden:zed-editor--1.19.2-->-1.20.1
Sep 16, 2026
Merged

GaetanLepage merged 1 commit into
NixOS:masterfrom
schembriaiden:zed-editor--1.19.2-->-1.20.1

Conversation

@schembriaiden

Copy link
Copy Markdown
Contributor

Changelog: https://github.com/zed-industries/zed/releases/tag/v1.20.1

Things done

@nixpkgs-ci nixpkgs-ci Bot added 8.has: package (update) This PR updates a package to a newer version 10.rebuild-linux: 1-10 This PR causes between 1 and 10 packages to rebuild on Linux. 10.rebuild-darwin: 1-10 This PR causes between 1 and 10 packages to rebuild on Darwin. 11.by: package-maintainer This PR was created by a maintainer of all the package it changes. 10.rebuild-darwin: 1 This PR causes 1 package to rebuild on Darwin. labels Sep 16, 2026
@nixpkgs-review-gha

Copy link
Copy Markdown

nixpkgs-review result

Generated using nixpkgs-review-gha (d5406bd)

Command: nixpkgs-review pr 563960
Commit: c3b9190ca7c2ee76e911844a98f2101e7a1688f8 (subsequent changes)
Merge: ea4efd392231a222ebd2fdef983092ddd1519cf9

Triggered by @schembriaiden (81376423)
Logs: https://github.com/schembriaiden/nixpkgs-review-gha/actions/runs/35123321979/attempts/1


x86_64-linux

✅ 2 packages built:
  • zed-editor
  • zed-editor-fhs

aarch64-linux

✅ 2 packages built:
  • zed-editor
  • zed-editor-fhs

aarch64-darwin (sandbox = relaxed)

✅ 1 package built:
  • zed-editor

@GaetanLepage
GaetanLepage added this pull request to the merge queue Sep 16, 2026
@nixpkgs-ci nixpkgs-ci Bot added 12.approvals: 1 This PR was reviewed and approved by one person. 12.approved-by: package-maintainer This PR was reviewed and approved by a maintainer listed in any of the changed packages. 2.status: merge-bot eligible This PR can be merged by commenting "@NixOS/nixpkgs-merge-bot merge". labels Sep 16, 2026
Merged via the queue into NixOS:master with commit 2ead8cb Sep 16, 2026
35 checks passed
@schembriaiden
schembriaiden deleted the zed-editor--1.19.2-->-1.20.1 branch September 16, 2026 19:36
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

2.status: merge-bot eligible This PR can be merged by commenting "@NixOS/nixpkgs-merge-bot merge". 8.has: package (update) This PR updates a package to a newer version 10.rebuild-darwin: 1-10 This PR causes between 1 and 10 packages to rebuild on Darwin. 10.rebuild-darwin: 1 This PR causes 1 package to rebuild on Darwin. 10.rebuild-linux: 1-10 This PR causes between 1 and 10 packages to rebuild on Linux. 11.by: package-maintainer This PR was created by a maintainer of all the package it changes. 12.approvals: 1 This PR was reviewed and approved by one person. 12.approved-by: package-maintainer This PR was reviewed and approved by a maintainer listed in any of the changed packages.

Projects

None yet

Development

Successfully merging this pull request may close these issues.

3 participants