Skip to content

pokego: 0.5.8 -> 0.5.9 - #563964

Merged
NickCao merged 1 commit into
NixOS:masterfrom
yvnth:pokego/0.5.9
Sep 16, 2026
Merged

NickCao merged 1 commit into
NixOS:masterfrom
yvnth:pokego/0.5.9

Conversation

@yvnth

@yvnth yvnth commented Sep 16, 2026

Copy link
Copy Markdown
Contributor

Change log: https://github.com/rubiin/pokego/releases/tag/v0.5.9
Diff: rubiin/pokego@v0.5.8...v0.5.9

Things done

@yvnth

yvnth commented Sep 16, 2026

Copy link
Copy Markdown
Contributor Author

nixpkgs-review result

Generated using nixpkgs-review-gha (d5406bd)

Command: nixpkgs-review pr 563964
Commit: dd16649fb05a623e4b7cbfa1527edb3ea281dbd5 (subsequent changes)
Merge: cc4669fc2a55dd87295b10e155a8a61fc0a8754c

Logs: https://github.com/yvnth/nixpkgs-review-gha-fork/actions/runs/35124846791/attempts/1


x86_64-linux

✅ 1 package built:
  • pokego

aarch64-linux

✅ 1 package built:
  • pokego

@nixpkgs-ci nixpkgs-ci Bot added 8.has: package (update) This PR updates a package to a newer version 10.rebuild-linux: 1-10 This PR causes between 1 and 10 packages to rebuild on Linux. 10.rebuild-darwin: 0 This PR does not cause any packages to rebuild on Darwin. 10.rebuild-linux: 1 This PR causes 1 package to rebuild on Linux. 9.needs: reviewer This PR currently has no reviewers requested and needs attention. labels Sep 16, 2026
@NickCao
NickCao added this pull request to the merge queue Sep 16, 2026
@nixpkgs-ci nixpkgs-ci Bot added 12.approvals: 1 This PR was reviewed and approved by one person. and removed 9.needs: reviewer This PR currently has no reviewers requested and needs attention. labels Sep 16, 2026
Merged via the queue into NixOS:master with commit 5b773f8 Sep 16, 2026
33 checks passed
@yvnth
yvnth deleted the pokego/0.5.9 branch September 16, 2026 20:56
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

8.has: package (update) This PR updates a package to a newer version 10.rebuild-darwin: 0 This PR does not cause any packages to rebuild on Darwin. 10.rebuild-linux: 1-10 This PR causes between 1 and 10 packages to rebuild on Linux. 10.rebuild-linux: 1 This PR causes 1 package to rebuild on Linux. 12.approvals: 1 This PR was reviewed and approved by one person.

Projects

None yet

Development

Successfully merging this pull request may close these issues.

2 participants