Skip to content

casadi: 3.8.0 -> 3.8.1 - #563966

Merged
nixpkgs-ci[bot] merged 1 commit into
NixOS:masterfrom
r-ryantm:auto-update/casadi
Sep 16, 2026
Merged

nixpkgs-ci[bot] merged 1 commit into
NixOS:masterfrom
r-ryantm:auto-update/casadi

Conversation

@r-ryantm

Copy link
Copy Markdown
Contributor

Automatic update generated by nixpkgs-update tools. This update was made based on information from https://github.com/casadi/casadi/releases.

meta.description for casadi is: Symbolic framework for numeric optimization

meta.homepage for casadi is: https://github.com/casadi/casadi

meta.changelog for casadi is: https://github.com/casadi/casadi/releases/tag/3.8.1

Updates performed
  • Version update
To inspect upstream changes
Impact

Checks done


  • built on NixOS
  • The tests defined in passthru.tests, if any, passed
  • found 3.8.1 with grep in /nix/store/45l2a794b31x9v6p49frzr7zc3d2acr7-casadi-3.8.1
  • found 3.8.1 in filename of file in /nix/store/45l2a794b31x9v6p49frzr7zc3d2acr7-casadi-3.8.1

Rebuild report (if merged into master) (click to expand)
31 total rebuild path(s)

31 package rebuild(s)

First fifty rebuilds by attrpath

aligator
casadi
crocoddyl
haskellPackages.casadi-bindings-internal
mim-solvers
ndcurves
pinocchio
pkgsRocm.python3Packages.mim-solvers
python313Packages.aerosandbox
python313Packages.aligator
python313Packages.casadi
python313Packages.crocoddyl
python313Packages.example-robot-data
python313Packages.mim-solvers
python313Packages.ndcurves
python313Packages.neuralfoil
python313Packages.pinocchio
python313Packages.tsid
python314Packages.aerosandbox
python314Packages.aligator
python314Packages.casadi
python314Packages.crocoddyl
python314Packages.example-robot-data
python314Packages.mim-solvers
python314Packages.ndcurves
python314Packages.neuralfoil
python314Packages.pinocchio
python314Packages.tsid
toppra
tsid
Instructions to test this update (click to expand)

Either download from the cache:

nix-store -r /nix/store/45l2a794b31x9v6p49frzr7zc3d2acr7-casadi-3.8.1 \
  --option binary-caches 'https://cache.nixos.org/ https://nixpkgs-update-cache.nix-community.org/' \
  --option trusted-public-keys '
  nixpkgs-update-cache.nix-community.org-1:U8d6wiQecHUPJFSqHN9GSSmNkmdiFW7GW7WNAnHW0SM=
  cache.nixos.org-1:6NCHdD59X431o0gWypbMrAURkbJ16ZPMQFGspcDShjY=
  '

(The nixpkgs-update cache is only trusted for this store-path realization.)
For the cached download to work, your user must be in the trusted-users list or you can use sudo since root is effectively trusted.

Or, build yourself:

nix-build -A casadi https://github.com/r-ryantm/nixpkgs/archive/ec65eb4e1d7ade195045fa67eb3e7c8965513536.tar.gz

Or:

nix build github:r-ryantm/nixpkgs/ec65eb4e1d7ade195045fa67eb3e7c8965513536#casadi

After you've downloaded or built it, look at the files and if there are any, run the binaries:

ls -la /nix/store/45l2a794b31x9v6p49frzr7zc3d2acr7-casadi-3.8.1
ls -la /nix/store/45l2a794b31x9v6p49frzr7zc3d2acr7-casadi-3.8.1/bin


Pre-merge build results

We have automatically built all packages that will get rebuilt due to
this change.

This gives evidence on whether the upgrade will break dependent packages.
Note sometimes packages show up as failed to build independent of the
change, simply because they are already broken on the target branch.

nixpkgs-review result

Generated using nixpkgs-review.

Command: nixpkgs-review --extra-nixpkgs-config '{ allowBroken = false; }'
Commit: ec65eb4e1d7ade195045fa67eb3e7c8965513536


x86_64-linux

❌ 20 packages failed to build:
  • python313Packages.aerosandbox
  • python313Packages.casadi
  • python313Packages.example-robot-data
  • python313Packages.crocoddyl
  • python313Packages.ndcurves
  • python313Packages.neuralfoil
  • python313Packages.pinocchio
  • python313Packages.tsid
  • python313Packages.aligator
  • python314Packages.aerosandbox
  • python314Packages.casadi
  • python314Packages.crocoddyl
  • python314Packages.example-robot-data
  • python314Packages.aligator
  • python314Packages.ndcurves
  • python313Packages.mim-solvers
  • python314Packages.pinocchio
  • python314Packages.neuralfoil
  • python314Packages.tsid
  • python314Packages.mim-solvers
✅ 9 packages built:
  • casadi
  • haskellPackages.casadi-bindings-internal
  • aligator
  • ndcurves
  • pinocchio
  • mim-solvers
  • crocoddyl
  • tsid
  • toppra

Maintainer pings

cc @nim65s for testing.

Tip

As a maintainer, if your package is located under pkgs/by-name/*, you can comment @NixOS/nixpkgs-merge-bot merge to automatically merge this update using the nixpkgs-merge-bot.

@nixpkgs-ci nixpkgs-ci Bot added the 2.status: merge-bot eligible This PR can be merged by commenting "@NixOS/nixpkgs-merge-bot merge". label Sep 16, 2026
@nixpkgs-ci
nixpkgs-ci Bot requested a review from nim65s September 16, 2026 17:02
@nixpkgs-ci nixpkgs-ci Bot added 8.has: package (update) This PR updates a package to a newer version 10.rebuild-linux: 11-100 This PR causes between 11 and 100 packages to rebuild on Linux. 10.rebuild-darwin: 11-100 This PR causes between 11 and 100 packages to rebuild on Darwin. labels Sep 16, 2026

@nim65s nim65s left a comment

Copy link
Copy Markdown
Contributor

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

changelog lgtm

@nim65s

nim65s commented Sep 16, 2026

Copy link
Copy Markdown
Contributor

@NixOS/nixpkgs-merge-bot merge

@nixpkgs-ci
nixpkgs-ci Bot added this pull request to the merge queue Sep 16, 2026
@nixpkgs-ci

nixpkgs-ci Bot commented Sep 16, 2026

Copy link
Copy Markdown
Contributor

@nim65s wants to merge this PR.

Requirements to merge this PR with @NixOS/nixpkgs-merge-bot merge:

  • ✅ PR targets a development branch.
  • ✅ PR touches only files of packages in pkgs/by-name/.
  • ✅ PR is at least one of:
  • ✅ PR is not a draft
  • ✅ PR is not blocked by a "changes requested" review from a committer.
  • ✅ nim65s is a member of @NixOS/nixpkgs-maintainers.
  • ✅ nim65s is a maintainer of all touched packages on the master branch.

✔️ Queued for merge (#306934)

@nixpkgs-ci nixpkgs-ci Bot added 12.approvals: 1 This PR was reviewed and approved by one person. 12.approved-by: package-maintainer This PR was reviewed and approved by a maintainer listed in any of the changed packages. labels Sep 16, 2026
Merged via the queue into NixOS:master with commit 6e87954 Sep 16, 2026
33 checks passed
@r-ryantm
r-ryantm deleted the auto-update/casadi branch September 17, 2026 00:46
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

2.status: merge-bot eligible This PR can be merged by commenting "@NixOS/nixpkgs-merge-bot merge". 8.has: package (update) This PR updates a package to a newer version 10.rebuild-darwin: 11-100 This PR causes between 11 and 100 packages to rebuild on Darwin. 10.rebuild-linux: 11-100 This PR causes between 11 and 100 packages to rebuild on Linux. 12.approvals: 1 This PR was reviewed and approved by one person. 12.approved-by: package-maintainer This PR was reviewed and approved by a maintainer listed in any of the changed packages.

Projects

None yet

Development

Successfully merging this pull request may close these issues.

2 participants