Skip to content

restate: 1.7.9 -> 1.7.10 - #563967

Merged
NickCao merged 1 commit into
NixOS:masterfrom
r-ryantm:auto-update/restate
Sep 16, 2026
Merged

NickCao merged 1 commit into
NixOS:masterfrom
r-ryantm:auto-update/restate

Conversation

@r-ryantm

Copy link
Copy Markdown
Contributor

Automatic update generated by nixpkgs-update tools. This update was made based on information from passthru.updateScript.

meta.description for restate is: Platform for developing distributed fault-tolerant applications

meta.homepage for restate is: https://restate.dev

meta.changelog for restate is: https://github.com/restatedev/restate/releases/tag/v1.7.10

Updates performed
  • Ran passthru.UpdateScript
To inspect upstream changes
Impact

Checks done


  • built on NixOS
  • The tests defined in passthru.tests, if any, passed
  • found 1.7.10 with grep in /nix/store/6hpvngc7r1m6g3ik0l3mikknxz5ky7x8-restate-1.7.10
  • found 1.7.10 in filename of file in /nix/store/6hpvngc7r1m6g3ik0l3mikknxz5ky7x8-restate-1.7.10

Rebuild report (if merged into master) (click to expand)
8 total rebuild path(s)

2 package rebuild(s)

First fifty rebuilds by attrpath

restate
Instructions to test this update (click to expand)

Either download from the cache:

nix-store -r /nix/store/6hpvngc7r1m6g3ik0l3mikknxz5ky7x8-restate-1.7.10 \
  --option binary-caches 'https://cache.nixos.org/ https://nixpkgs-update-cache.nix-community.org/' \
  --option trusted-public-keys '
  nixpkgs-update-cache.nix-community.org-1:U8d6wiQecHUPJFSqHN9GSSmNkmdiFW7GW7WNAnHW0SM=
  cache.nixos.org-1:6NCHdD59X431o0gWypbMrAURkbJ16ZPMQFGspcDShjY=
  '

(The nixpkgs-update cache is only trusted for this store-path realization.)
For the cached download to work, your user must be in the trusted-users list or you can use sudo since root is effectively trusted.

Or, build yourself:

nix-build -A restate https://github.com/r-ryantm/nixpkgs/archive/fe73660b1ab7df01a004d9943999f738e47532b9.tar.gz

Or:

nix build github:r-ryantm/nixpkgs/fe73660b1ab7df01a004d9943999f738e47532b9#restate

After you've downloaded or built it, look at the files and if there are any, run the binaries:

ls -la /nix/store/6hpvngc7r1m6g3ik0l3mikknxz5ky7x8-restate-1.7.10
ls -la /nix/store/6hpvngc7r1m6g3ik0l3mikknxz5ky7x8-restate-1.7.10/bin


Pre-merge build results

We have automatically built all packages that will get rebuilt due to
this change.

This gives evidence on whether the upgrade will break dependent packages.
Note sometimes packages show up as failed to build independent of the
change, simply because they are already broken on the target branch.

nixpkgs-review result

Generated using nixpkgs-review.

Command: nixpkgs-review --extra-nixpkgs-config '{ allowBroken = false; }'
Commit: fe73660b1ab7df01a004d9943999f738e47532b9


x86_64-linux

✅ 1 package built:
  • restate

Maintainer pings

cc @myypo for testing.

Tip

As a maintainer, if your package is located under pkgs/by-name/*, you can comment @NixOS/nixpkgs-merge-bot merge to automatically merge this update using the nixpkgs-merge-bot.

@nixpkgs-ci nixpkgs-ci Bot added the 2.status: merge-bot eligible This PR can be merged by commenting "@NixOS/nixpkgs-merge-bot merge". label Sep 16, 2026
@nixpkgs-ci
nixpkgs-ci Bot requested a review from myypo September 16, 2026 17:04
@nixpkgs-ci nixpkgs-ci Bot added 8.has: package (update) This PR updates a package to a newer version 10.rebuild-linux: 1-10 This PR causes between 1 and 10 packages to rebuild on Linux. 10.rebuild-darwin: 1-10 This PR causes between 1 and 10 packages to rebuild on Darwin. 10.rebuild-darwin: 1 This PR causes 1 package to rebuild on Darwin. 10.rebuild-linux: 1 This PR causes 1 package to rebuild on Linux. labels Sep 16, 2026
@NickCao
NickCao added this pull request to the merge queue Sep 16, 2026
@nixpkgs-ci nixpkgs-ci Bot added the 12.approvals: 1 This PR was reviewed and approved by one person. label Sep 16, 2026
Merged via the queue into NixOS:master with commit cdd7ed6 Sep 16, 2026
34 checks passed
@r-ryantm
r-ryantm deleted the auto-update/restate branch September 17, 2026 00:46
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

2.status: merge-bot eligible This PR can be merged by commenting "@NixOS/nixpkgs-merge-bot merge". 8.has: package (update) This PR updates a package to a newer version 10.rebuild-darwin: 1-10 This PR causes between 1 and 10 packages to rebuild on Darwin. 10.rebuild-darwin: 1 This PR causes 1 package to rebuild on Darwin. 10.rebuild-linux: 1-10 This PR causes between 1 and 10 packages to rebuild on Linux. 10.rebuild-linux: 1 This PR causes 1 package to rebuild on Linux. 12.approvals: 1 This PR was reviewed and approved by one person.

Projects

None yet

Development

Successfully merging this pull request may close these issues.

2 participants