Skip to content

v3.21.0 - #387

Merged
n9d0g merged 2 commits into
mainfrom
dev-web
Jul 3, 2026
Merged

v3.21.0#387
n9d0g merged 2 commits into
mainfrom
dev-web

Conversation

@n9d0g

@n9d0g n9d0g commented Jul 3, 2026

Copy link
Copy Markdown
Owner

No description provided.

@n9d0g n9d0g self-assigned this Jul 3, 2026
@changeset-bot

changeset-bot Bot commented Jul 3, 2026

Copy link
Copy Markdown

⚠️ No Changeset found

Latest commit: a2e886c

Merging this PR will not cause a version bump for any packages. If these changes should not result in a new version, you're good to go. If these changes should result in a version bump, you need to add a changeset.

This PR includes no changesets

When changesets are added to this PR, you'll see the packages that this PR includes changesets for and the associated semver types

Click here to learn what changesets are, and how to add one.

Click here if you're a maintainer who wants to add a changeset to this PR

@vercel

vercel Bot commented Jul 3, 2026

Copy link
Copy Markdown

The latest updates on your projects. Learn more about Vercel for GitHub.

Project Deployment Actions Updated (UTC)
fcc Ready Ready Preview, Comment Jul 3, 2026 6:08pm
1 Skipped Deployment
Project Deployment Actions Updated (UTC)
fcc-cms Ignored Ignored Jul 3, 2026 6:08pm

@n9d0g n9d0g changed the title Dev web v3.21.0 Jul 3, 2026
@n9d0g n9d0g linked an issue Jul 3, 2026 that may be closed by this pull request
@n9d0g
n9d0g merged commit 4ee5162 into main Jul 3, 2026
4 checks passed
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Projects

None yet

Development

Successfully merging this pull request may close these issues.

typescript any fixes

1 participant